Skip to content

HADOOP-19990: Upgrade to Bouncycastle 1.86 due to CVEs - #8743

Open
pjfanning wants to merge 1 commit into
apache:trunkfrom
pjfanning:HADOOP-19990-bc
Open

pjfanning wants to merge 1 commit into
apache:trunkfrom
pjfanning:HADOOP-19990-bc

Conversation

@pjfanning

Copy link
Copy Markdown
Member

Description of PR

https://issues.apache.org/jira/browse/HADOOP-19990

How was this patch tested?

For code changes:

  • Does the title of this PR start with the corresponding JIRA issue id (e.g. 'HADOOP-17799. Your PR title ...')?
  • Object storage: Have the integration tests been executed and the endpoint
    declared according to the connector-specific documentation? Note: Automated CI
    testing doesn't cover all cases so manual testing with cloud storage is still
    required.
  • If adding new dependencies to the code, are these dependencies licensed in a way that is compatible for inclusion under ASF 2.0?
  • If applicable, have you updated the LICENSE, LICENSE-binary, NOTICE-binary files?

AI Tooling

If an AI tool was used:

@hadoop-yetus

Copy link
Copy Markdown

💔 -1 overall

Vote Subsystem Runtime Logfile Comment
+0 🆗 reexec 13m 4s Docker mode activated.
_ Prechecks _
+1 💚 dupname 0m 0s No case conflicting files found.
+0 🆗 codespell 0m 0s codespell was not available.
+0 🆗 detsecrets 0m 0s detect-secrets was not available.
+0 🆗 xmllint 0m 0s xmllint was not available.
+0 🆗 shelldocs 0m 0s Shelldocs was not available.
+1 💚 @author 0m 0s The patch does not contain any @author tags.
-1 ❌ test4tests 0m 0s The patch doesn't appear to include any new or modified tests. Please justify why no new tests are needed for this patch. Also please list what manual steps were performed to verify this patch.
_ trunk Compile Tests _
+0 🆗 mvndep 2m 9s Maven dependency ordering for branch
+1 💚 mvninstall 43m 36s trunk passed
+1 💚 compile 16m 10s trunk passed with JDK Ubuntu-21.0.12+8-1-24.04-Ubuntu
+1 💚 compile 17m 55s trunk passed with JDK Ubuntu-17.0.20+8-1-24.04-Ubuntu
+1 💚 mvnsite 19m 8s trunk passed
+1 💚 javadoc 9m 55s trunk passed with JDK Ubuntu-21.0.12+8-1-24.04-Ubuntu
+1 💚 javadoc 9m 42s trunk passed with JDK Ubuntu-17.0.20+8-1-24.04-Ubuntu
+1 💚 shadedclient 47m 28s branch has no errors when building and testing our client artifacts.
_ Patch Compile Tests _
+0 🆗 mvndep 0m 32s Maven dependency ordering for patch
+1 💚 mvninstall 36m 28s the patch passed
+1 💚 compile 16m 2s the patch passed with JDK Ubuntu-21.0.12+8-1-24.04-Ubuntu
+1 💚 javac 16m 2s the patch passed
+1 💚 compile 16m 36s the patch passed with JDK Ubuntu-17.0.20+8-1-24.04-Ubuntu
+1 💚 javac 16m 36s the patch passed
+1 💚 blanks 0m 0s The patch has no blanks issues.
+1 💚 mvnsite 18m 36s the patch passed
+1 💚 shellcheck 0m 0s No new issues.
+1 💚 javadoc 9m 49s the patch passed with JDK Ubuntu-21.0.12+8-1-24.04-Ubuntu
+1 💚 javadoc 9m 39s the patch passed with JDK Ubuntu-17.0.20+8-1-24.04-Ubuntu
+1 💚 shadedclient 49m 2s patch has no errors when building and testing our client artifacts.
_ Other Tests _
-1 ❌ unit 769m 46s /patch-unit-root.txt root in the patch failed.
+1 💚 asflicense 1m 59s The patch does not generate ASF License warnings.
1072m 57s
Reason Tests
Failed junit tests hadoop.yarn.service.TestYarnNativeServices
hadoop.yarn.server.router.webapp.TestFederationWebApp
hadoop.yarn.server.resourcemanager.reservation.TestCapacityOverTimePolicy
hadoop.yarn.server.nodemanager.containermanager.logaggregation.TestLogAggregationService
Subsystem Report/Notes
Docker ClientAPI=1.56 ServerAPI=1.56 base: https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-8743/1/artifact/out/Dockerfile
GITHUB PR #8743
Optional Tests dupname asflicense compile javac javadoc mvninstall mvnsite unit shadedclient codespell detsecrets xmllint shellcheck shelldocs
uname Linux ecedf6c15cbc 5.15.0-190-generic #200-Ubuntu SMP Fri Aug 7 15:06:04 UTC 2026 x86_64 x86_64 x86_64 GNU/Linux
Build tool maven
Personality dev-support/bin/hadoop.sh
git revision trunk / 1359caa
Default Java Ubuntu-17.0.20+8-1-24.04-Ubuntu
Multi-JDK versions /usr/lib/jvm/java-21-openjdk-amd64:Ubuntu-21.0.12+8-1-24.04-Ubuntu /usr/lib/jvm/java-17-openjdk-amd64:Ubuntu-17.0.20+8-1-24.04-Ubuntu
Test Results https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-8743/1/testReport/
Max. process+thread count 4054 (vs. ulimit of 10000)
modules C: hadoop-project . U: .
Console output https://ci-hadoop.apache.org/job/hadoop-multibranch/job/PR-8743/1/console
versions git=2.43.0 maven=3.9.15 shellcheck=0.9.0
Powered by Apache Yetus 0.14.1 https://yetus.apache.org

This message was automatically generated.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants