Skip to content

Enables the use of dynamic client - #1035

Merged
k8s-ci-robot merged 1 commit into
kubernetes-client:masterfrom
micw523:enable-dynamic
Jan 6, 2020
Merged

k8s-ci-robot merged 1 commit into
kubernetes-client:masterfrom
micw523:enable-dynamic

Conversation

@micw523

@micw523 micw523 commented Dec 18, 2019

Copy link
Copy Markdown
Contributor

When I'm switching our create_from_yaml utility to the dynamic client to allow easier treatment of the CRDs, I found out that the dynamic client was never imported into the package and cannot be used. This PR enables the users to actually use the dynamic client.

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Dec 18, 2019
@k8s-ci-robot k8s-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Dec 18, 2019
@roycaihw

roycaihw commented Jan 6, 2020

Copy link
Copy Markdown
Member

/cc @fabianvf
/assign

@roycaihw

roycaihw commented Jan 6, 2020

Copy link
Copy Markdown
Member

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 6, 2020
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: micw523, roycaihw

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 6, 2020
@k8s-ci-robot
k8s-ci-robot merged commit 96766c2 into kubernetes-client:master Jan 6, 2020
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants