Repository navigation
Update urllib3 for CVE-2019-11324 - #897
Merged
Merged
Conversation
Contributor
Author
Member
|
/lgtm @yliaog I'm not certain if we need to cherrypick this |
Contributor
|
/lgtm CVE-2019-11324 is deemed high impact, so better to have a new release with the fix soon |
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: micw523, yliaog The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Member
|
@yliaog sounds good. I will do releases for k8s 1.13, 1.14 and 1.15 |
wmfgerrit
pushed a commit
to wikimedia/labs-tools-maintain-kubeusers
that referenced
this pull request
Oct 31, 2019
This will cover kubernetes-client/python#897 Also included are upgrades to the cryptography package and a bugfix update to ldap3. Tests, builds and full runs against test environments completed successfully. Change-Id: Ica345d092f2ce5c59cdd9f4e7144ae22b594dc6b
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
CVE-2019-11324 affects all urllib3 versions before 1.24.2. We should bump the minimum version required.