Skip to content

Request 2 CPUs per Travis CI job - #906

Closed
micw523 wants to merge 1 commit into
kubernetes-client:masterfrom
micw523:patch-4
Closed

micw523 wants to merge 1 commit into
kubernetes-client:masterfrom
micw523:patch-4

Conversation

@micw523

@micw523 micw523 commented Jul 30, 2019

Copy link
Copy Markdown
Contributor

Add an environmental variable according to this help to avoid test flakes

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jul 30, 2019
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: micw523
To complete the pull request process, please assign yliaog
You can assign the PR to them by writing /assign @yliaog in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jul 30, 2019
@micw523

micw523 commented Jul 30, 2019

Copy link
Copy Markdown
Contributor Author

/hold
Need to see if CI likes it

@k8s-ci-robot k8s-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jul 30, 2019
@micw523

micw523 commented Jul 30, 2019

Copy link
Copy Markdown
Contributor Author

/hold cancel
/assign @roycaihw
Looks OK on my personal build.

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jul 30, 2019
Comment thread scripts/kube-init.sh
setenforce 0

# Request at least 2 CPUs from Travis CI per job
export TRAVIS_WORKER_DOCKER_CPUS=2

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I did some search and it seems the default value for TRAVIS_WORKER_DOCKER_CPUS is 2 already. I'm not sure if this is the right fix

https://github.com/travis-ci/worker/blob/b6839eb59412dd46ec2feccc85660c2afd990772/.example.env#L49-L51
https://github.com/travis-ci/worker/blob/b6839eb59412dd46ec2feccc85660c2afd990772/backend/docker.go#L60

For the original issue #890, have you observed other builds in this repo hitting the same error?

@micw523

micw523 commented Jul 31, 2019

Copy link
Copy Markdown
Contributor Author

I also think the default is 2. The official Travis CI documentation also says at least 2 CPUs are needed to run the jobs. Nothing started flaking until a few days ago, when only 1 CPU is assigned to some of our jobs. Today’s and yesterday’s builds are fine though - I wonder if something happened on the Travis side

@micw523

micw523 commented Jul 31, 2019

Copy link
Copy Markdown
Contributor Author

The most recent one is from #795. For some others that were failing, I restarted the individual py-functional jobs a few times and they went green - I’m not sure if we have those records...

@micw523

micw523 commented Jul 31, 2019

Copy link
Copy Markdown
Contributor Author

Ok so it’s documented here - https://travis-ci.community/t/builds-often-fail-because-of-lack-of-resources-memory-cpu/4331/3

The failure is due to Travis CI.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants