Repository navigation
chore(deps): bump github.com/letsencrypt/boulder from 0.20260908.0 to 0.20260928.1 - #3327
Open
dependabot[bot] wants to merge 1 commit into
Conversation
Bumps [github.com/letsencrypt/boulder](https://github.com/letsencrypt/boulder) from 0.20260908.0 to 0.20260928.1. - [Release notes](https://github.com/letsencrypt/boulder/releases) - [Changelog](https://github.com/letsencrypt/boulder/blob/main/docs/release.md) - [Commits](letsencrypt/boulder@v0.20260908.0...v0.20260928.1) --- updated-dependencies: - dependency-name: github.com/letsencrypt/boulder dependency-version: 0.20260928.1 dependency-type: direct:production update-type: version-update:semver-minor ... Signed-off-by: dependabot[bot] <support@github.com>
Contributor
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Member
|
/retest |
Member
|
/hold |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Bumps github.com/letsencrypt/boulder from 0.20260908.0 to 0.20260928.1.
Release notes
Sourced from github.com/letsencrypt/boulder's releases.
Commits
7a9802dDon't build releases on go1.26.1 (#9066)74a4914linter: deprecate UnsignLintCerts flag (#9059)174cba8wfe: 404 for malformed and <= 0 path components (#9042)4f82920admin: invert reasonCode check when -skip-block-key is supplied (#9058)ecfc386admin: require explicit reasonCode for revoke-cert commands (#9057)9c6cd1dci: update sunlight and vitess versions (#9056)884a968ci: add seaweedfs as an S3 backend (#9041)77d0cf6Add lint to check that Root CRLs contain reason codes for all entries (#9049)29a3603ceremony: use RevocationListEntry.ReasonCode (#9045)da514c9build(deps): bump the aws group with 4 updates (#9043)Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)